Sitelet https://github.com/kubernetes/kops/pull/18482
Skip to content

API only CLI - #18482

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
cheftako:apiOnlyCli
Jun 18, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
cheftako:apiOnlyCli

Conversation

@cheftako

Copy link
Copy Markdown
Member

Added an option to provide apiserver only machine type in the cli.
Added test for the generated spec from this.
Fixed comment from previous pr.

@k8s-ci-robot k8s-ci-robot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Jun 17, 2026
@k8s-ci-robot
k8s-ci-robot requested review from hakman and zetaab June 17, 2026 21:09
@k8s-ci-robot k8s-ci-robot added area/api area/documentation cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Jun 17, 2026
@cheftako

cheftako commented Jun 17, 2026 •

Copy link
Copy Markdown
Member Author

/test pull-kops-scenario-splitkcp-gcp

@cheftako

Copy link
Copy Markdown
Member Author

/test pull-kops-scenario-clusterapi-gcp

@hakman hakman changed the title Api only cli Add API only cli Jun 18, 2026
@hakman hakman changed the title Add API only cli Add API only CLI Jun 18, 2026
@cheftako

Copy link
Copy Markdown
Member Author

/test pull-kops-scenario-clusterapi-gcp

@cheftako cheftako changed the title Add API only CLI [WIP] Api only cli Jun 18, 2026
@k8s-ci-robot k8s-ci-robot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Jun 18, 2026
@cheftako
cheftako force-pushed the apiOnlyCli branch 2 times, most recently from 21489be to d0a4abc Compare June 18, 2026 18:19
Added flag --api-server-size to be consistent with other machine type
flags.
Added doc on the flag reflecting my testing.
Adding GCE test for APIServer only option.
Fixed comment from previous PR.
apiserver only DNS check for AWS comment is now correct.
Removed k8s version flag from doc.
make gen-cli-docs
@cheftako cheftako changed the title [WIP] Api only cli Api only cli Jun 18, 2026
@k8s-ci-robot k8s-ci-robot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Jun 18, 2026
Comment on lines +59 to +81
# Install extra RBAC for kops-controller CAPI support
kubectl apply --server-side -f "${REPO_ROOT}/clusterapi/examples/kopscontroller.yaml"

# Bounce kops-controller in case it went into backoff before the CRDs were installed
kubectl delete pod -n kube-system -l k8s-app=kops-controller

# Install cert-manager
kubectl apply --server-side -f https://github.com/cert-manager/cert-manager/releases/download/v1.18.2/cert-manager.yaml

kubectl wait --for=condition=Available --timeout=5m -n cert-manager deployment/cert-manager
kubectl wait --for=condition=Available --timeout=5m -n cert-manager deployment/cert-manager-cainjector
kubectl wait --for=condition=Available --timeout=5m -n cert-manager deployment/cert-manager-webhook

# Install cluster-api core and cluster-api-provider-gcp
kubectl apply --server-side -k "${REPO_ROOT}/clusterapi/manifests/cluster-api"
kubectl wait --for=condition=Available --timeout=5m -n capi-system deployment/capi-controller-manager

kubectl apply --server-side -k "${REPO_ROOT}/clusterapi/manifests/cluster-api-provider-gcp"
kubectl wait --for=condition=Available --timeout=5m -n capg-system deployment/capg-controller-manager


# Install extra RBAC for capi-manager loopback connection to cluster (used to check node health etc)
kubectl apply --server-side -f "${REPO_ROOT}/clusterapi/examples/capi-loopback.yaml"

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Do we want to keep doing CAPI or shall we just bring up a cluster with dedicated apiserver nodes and make sure that we see the apiserver nodes? (It might need a go test in this directory, which I'm OK with)

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I'd like to turn this into its own real test. Not done allowing the KCP to be split. As we enable further splits I would like this become the test that all the pieces work.

g.Spec.MachineType = size
}
if i > 0 {
klog.Fatalf("multiple machine types for IG group not currently supported")

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Nit: "for apiserver InstanceGroup" or similar

@@ -0,0 +1,104 @@
#!/usr/bin/env bash

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

When we get this merged, let's create a PR to test-infra to make sure we run the scenario

@justinsb

Copy link
Copy Markdown
Member

One nit on message naming, one suggestion for future direction, but not blocking (tests).

/approve
/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 18, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: justinsb

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 18, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 1fcef80 into kubernetes:master Jun 18, 2026
27 checks passed
@hakman hakman changed the title Api only cli API only CLI Jun 19, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/api area/documentation cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants