Sitelet https://github.com/kubernetes/kops/pull/18480
Skip to content

channels: Add Kubernetes 1.36.0 support - #18480

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:channels-k8s-1.36
Jun 17, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:channels-k8s-1.36

Conversation

@hakman

@hakman hakman commented Jun 17, 2026

Copy link
Copy Markdown
Member

@k8s-ci-robot
k8s-ci-robot requested review from ameukam and rifelpet June 17, 2026 06:21
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. area/channels labels Jun 17, 2026
@ameukam

ameukam commented Jun 17, 2026

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 17, 2026
@hakman

hakman commented Jun 17, 2026

Copy link
Copy Markdown
Member Author

/override pull-kops-e2e-k8s-gce-cilium
/approve

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-k8s-gce-cilium

Details

In response to this:

/override pull-kops-e2e-k8s-gce-cilium
/approve

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 17, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 795502a into kubernetes:master Jun 17, 2026
26 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/channels cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants