Sitelet https://github.com/kubernetes/kops/pull/18477
Skip to content

chore(channels): bump k8s versions in alpha channel - #18477

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
moshevayner:alpha-updates-2026-06-16
Jun 17, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
moshevayner:alpha-updates-2026-06-16

Conversation

@moshevayner

Copy link
Copy Markdown
Member

Signed-off-by: Moshe Vayner moshe@vayner.me

What this PR does / why we need it:

Which issue(s) this PR fixes (optional, in fixes #<issue number>(, fixes #<issue_number>, ...) format, will close the issue(s) when PR gets merged):

Special notes for your reviewer:

Signed-off-by: Moshe Vayner <moshe@vayner.me>
@k8s-ci-robot k8s-ci-robot added size/S Denotes a PR that changes 10-29 lines, ignoring generated files. area/channels labels Jun 17, 2026
@k8s-ci-robot
k8s-ci-robot requested review from hakman and olemarkus June 17, 2026 03:00
@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Jun 17, 2026

@hakman hakman left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks! 🙂

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 17, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 17, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 3089081 into kubernetes:master Jun 17, 2026
26 checks passed
@moshevayner
moshevayner deleted the alpha-updates-2026-06-16 branch June 17, 2026 04:08
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/channels cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants