Sitelet https://github.com/kubernetes/kops/pull/18475
Skip to content

Add serathius as scalability test owner - #18475

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
Jefftree:scalability-owners-serathius
Jun 16, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
Jefftree:scalability-owners-serathius

Conversation

@Jefftree

Copy link
Copy Markdown
Member

The kops scalability test run is owned by SIG Scalability, so this directory should have a SIG Scalability owner. Adds serathius to approvers and reviewers of tests/e2e/scenarios/scalability/OWNERS.

@Jefftree

Copy link
Copy Markdown
Member Author

/assign @serathius @dims

@k8s-ci-robot k8s-ci-robot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Jun 16, 2026
@k8s-ci-robot
k8s-ci-robot requested review from dims and hakman June 16, 2026 19:12
@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Jun 16, 2026
@dims

dims commented Jun 16, 2026

Copy link
Copy Markdown
Member

/approve
/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 16, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dims

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 16, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 17001fb into kubernetes:master Jun 16, 2026
27 checks passed
@Jefftree
Jefftree deleted the scalability-owners-serathius branch June 26, 2026 03:36
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants