Sitelet https://github.com/kubernetes/kops/pull/18468
Skip to content

gce: Reconcile HTTP health check changes on existing health checks - #18468

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:gce-httphealthcheck-reconcile
Jun 14, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:gce-httphealthcheck-reconcile

Conversation

@hakman

@hakman hakman commented Jun 13, 2026

Copy link
Copy Markdown
Member

Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@k8s-ci-robot
k8s-ci-robot requested review from ameukam and rifelpet June 13, 2026 06:02
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. area/provider/gcp Issues or PRs related to gcp provider labels Jun 13, 2026
@ameukam

ameukam commented Jun 13, 2026

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 13, 2026
@hakman

hakman commented Jun 14, 2026

Copy link
Copy Markdown
Member Author

/approve

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 14, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 05a23f8 into kubernetes:master Jun 14, 2026
25 of 26 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/provider/gcp Issues or PRs related to gcp provider cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants