Sitelet https://github.com/kubernetes/kops/pull/18467
Skip to content

aws: Reconcile target group health check changes on existing target groups - #18467

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:aws-reconcile-tg-health-check
Jun 14, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:aws-reconcile-tg-health-check

Conversation

@hakman

@hakman hakman commented Jun 13, 2026

Copy link
Copy Markdown
Member

https://testgrid.k8s.io/kops-presubmits#pull-kops-aws-upgrade-k135-ko135-to-k136-kolatest-many-addons

+ /home/prow/go/src/k8s.io/kops/.build/dist/linux/amd64/kops update cluster
Using cluster from kubectl context: pr18456-kops-aws-upgrade-k135-ko135-to-k136-kolatest-many-addons.tests-kops-aws.k8s.io

W0608 07:18:59.930646   54942 kopsassets.go:66] Using base url from env var: KOPS_BASE_URL="https://storage.googleapis.com/k8s-staging-kops/pulls/pull-kops-aws-upgrade-k135-ko135-to-k136-kolatest-many-addons/pull-87cf55318f80c8564de8cad695e50a725ae2c6c5/1.36.0-alpha.2+v1.36.0-alpha.1-58-g6a857af16a"
Will modify resources:
  TargetGroup/kops-controller-pr18456-k-dt36ke
  	HealthCheckProtocol 	 TCP -> HTTPS
  	HealthCheckPath     	 <nil> -> /healthz

Must specify --yes to apply changes

/cc @rifelpet @ameukam

@k8s-ci-robot
k8s-ci-robot requested review from ameukam and rifelpet June 13, 2026 05:37
@k8s-ci-robot k8s-ci-robot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. area/provider/aws Issues or PRs related to aws provider cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Jun 13, 2026
@kubernetes kubernetes deleted a comment from k8s-ci-robot Jun 13, 2026
@hakman

hakman commented Jun 13, 2026

Copy link
Copy Markdown
Member Author

/test pull-kops-aws-upgrade-k134-ko134-to-k136-kolatest-many-addons
/test pull-kops-aws-upgrade-k135-ko135-to-k136-kolatest-many-addons

@hakman
hakman force-pushed the aws-reconcile-tg-health-check branch from ba48c72 to 00a1286 Compare June 13, 2026 06:37
…roups

Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@hakman

hakman commented Jun 13, 2026

Copy link
Copy Markdown
Member Author

/test pull-kops-aws-upgrade-k134-ko134-to-k136-kolatest-many-addons
/test pull-kops-aws-upgrade-k135-ko135-to-k136-kolatest-many-addons

@ameukam

ameukam commented Jun 13, 2026

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 13, 2026
@hakman

hakman commented Jun 14, 2026

Copy link
Copy Markdown
Member Author

/approve

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 14, 2026
k8s-ci-robot added a commit that referenced this pull request Jun 14, 2026
…-upstream-release-1.36

Automated cherry pick of #18467: aws: Reconcile target group health check changes on existing target groups
@hakman

hakman commented Jun 14, 2026

Copy link
Copy Markdown
Member Author

/retest

@k8s-ci-robot
k8s-ci-robot merged commit 1c57cb2 into kubernetes:master Jun 14, 2026
28 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/provider/aws Issues or PRs related to aws provider cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants