Sitelet https://github.com/kubernetes/kops/pull/18458
Skip to content

Support etcd 3.7.0-rc.0 and allow overriding the etcd version in the scalability scenario - #18458

Merged
k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
Jefftree:add-etcd-3.7.0-rc.0
Jun 12, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
Jefftree:add-etcd-3.7.0-rc.0

Conversation

@Jefftree

@Jefftree Jefftree commented Jun 8, 2026 •

Copy link
Copy Markdown
Member

Adds etcd 3.7.0-rc.0 as a supported version and an optional ETCD_VERSION override in the scalability scenario so scale jobs can test rc candidate.

Opted not to blanket default the 3.7 for all scalability jobs as this is rc candidate, but maybe we should for proper soak.

We are soaking etcd 3.7 ahead of the release. The best way to do that is to integrate with kubernetes scalability test and ensure that we do not find regressions/bugs. Kubernetes is already using the 3.7-rc client library (ref),
The timeline is roughly

  • June 1: Release etcd 3.7-rc.0
  • July 1: Release etcd 3.7 and switch all CI to 3.7
  • End of July: Release Kubernetes 1.37

@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels Jun 8, 2026
@k8s-ci-robot
k8s-ci-robot requested review from hakman and olemarkus June 8, 2026 19:29
@Jefftree

Jefftree commented Jun 8, 2026

Copy link
Copy Markdown
Member Author

/cc @serathius

@k8s-ci-robot
k8s-ci-robot requested a review from serathius June 8, 2026 19:30
Jefftree added 2 commits June 8, 2026 15:35
…ity scenario

Signed-off-by: Jefftree <jeffrey.ying86@live.com>
Signed-off-by: Jefftree <jeffrey.ying86@live.com>
@Jefftree
Jefftree force-pushed the add-etcd-3.7.0-rc.0 branch from 66db949 to d9af2f8 Compare June 8, 2026 19:35
@hakman

hakman commented Jun 8, 2026

Copy link
Copy Markdown
Member

@Jefftree please explain why you need this in the description. It is unusual to add support for unreleased etcd versions.

@Jefftree

Jefftree commented Jun 8, 2026

Copy link
Copy Markdown
Member Author

Added a block above. We want to soak our scalability tests to use the release candidate ahead of the release. I didn't touch any of the default version selection (still 3.6) as this is a release candidate and should strictly be opt in for the scalability tests.

We are soaking etcd 3.7 ahead of the release. The best way to do that is to integrate with kubernetes scalability test and ensure that we do not find regressions/bugs. Kubernetes is already using the 3.7-rc client library (ref),
The timeline is roughly

  • June 1: Release etcd 3.7-rc.0
  • July 1: Release etcd 3.7 and switch all CI to 3.7
  • End of July: Release Kubernetes 1.37

@hakman

hakman commented Jun 8, 2026

Copy link
Copy Markdown
Member

OK, it looks good, but will have to wait until we cut the 1.36 branch. It should happen until the end of the week.
/hold

@k8s-ci-robot k8s-ci-robot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jun 8, 2026
@Jefftree

Jefftree commented Jun 8, 2026

Copy link
Copy Markdown
Member Author

sounds good, thank you!

@hakman

hakman commented Jun 12, 2026

Copy link
Copy Markdown
Member

/lgtm
/approve
/unhold

@k8s-ci-robot k8s-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jun 12, 2026
@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 12, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 12, 2026
@k8s-ci-robot
k8s-ci-robot merged commit c93af60 into kubernetes:master Jun 12, 2026
29 checks passed
@Jefftree
Jefftree deleted the add-etcd-3.7.0-rc.0 branch June 26, 2026 03:36
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants