Sitelet https://github.com/kubernetes/kops/pull/18456
Skip to content

gVisor: add HasGVisor() helper function - #18456

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:gvisor-helper
Jun 8, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:gvisor-helper

Conversation

@hakman

@hakman hakman commented Jun 8, 2026

Copy link
Copy Markdown
Member

@k8s-ci-robot
k8s-ci-robot requested review from ameukam and rifelpet June 8, 2026 06:07
@k8s-ci-robot k8s-ci-robot added area/api cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Jun 8, 2026
@hakman

hakman commented Jun 8, 2026

Copy link
Copy Markdown
Member Author

/approve

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 8, 2026
@ameukam

ameukam commented Jun 8, 2026

Copy link
Copy Markdown
Member

Thank you!
/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 8, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 2f7a60d into kubernetes:master Jun 8, 2026
28 checks passed
@hakman
hakman deleted the gvisor-helper branch June 8, 2026 08:32
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/api cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants