Sitelet https://github.com/kubernetes/kops/pull/18455
Skip to content

kube-proxy: assert buildPod command in unit test - #18455

Merged
k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
hakman:kube-proxy-build-pod-test
Jun 8, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
hakman:kube-proxy-build-pod-test

Conversation

@hakman

@hakman hakman commented Jun 8, 2026

Copy link
Copy Markdown
Member

@k8s-ci-robot
k8s-ci-robot requested review from ameukam and rifelpet June 8, 2026 05:52
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. area/nodeup labels Jun 8, 2026
@hakman
hakman force-pushed the kube-proxy-build-pod-test branch from 11cad45 to 73092f1 Compare June 8, 2026 06:02
@k8s-ci-robot k8s-ci-robot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. and removed size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Jun 8, 2026
@hakman

hakman commented Jun 8, 2026

Copy link
Copy Markdown
Member Author

/approve

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 8, 2026
@ameukam

ameukam commented Jun 8, 2026

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 8, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 0229895 into kubernetes:master Jun 8, 2026
26 checks passed
@hakman
hakman deleted the kube-proxy-build-pod-test branch June 8, 2026 07:09
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/nodeup cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants