Sitelet https://github.com/kubernetes/kops/pull/18447
Skip to content

chore: downgrade containerd to v2.2.4 - #18447

Merged
k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
hakman:downgrade-containerd
Jun 5, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
hakman:downgrade-containerd

Conversation

@hakman

@hakman hakman commented Jun 5, 2026

Copy link
Copy Markdown
Member

containerd/containerd#13529 breaks kOps, and would rather not do some weird workaround.

/cc @rifelpet @ameukam

@k8s-ci-robot
k8s-ci-robot requested review from ameukam and rifelpet June 5, 2026 03:22
@k8s-ci-robot k8s-ci-robot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. area/documentation labels Jun 5, 2026
@hakman

hakman commented Jun 5, 2026

Copy link
Copy Markdown
Member Author

/retest

@hakman

hakman commented Jun 5, 2026

Copy link
Copy Markdown
Member Author

/override pull-kops-e2e-gce-cni-cilium-etcd
/approve

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-gce-cni-cilium-etcd

Details

In response to this:

/override pull-kops-e2e-gce-cni-cilium-etcd
/approve

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 5, 2026
@ameukam

ameukam commented Jun 5, 2026

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 5, 2026
@k8s-ci-robot
k8s-ci-robot merged commit fbd3406 into kubernetes:master Jun 5, 2026
40 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/documentation cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants