Sitelet https://github.com/kubernetes/kops/pull/18437
Skip to content

Cherry pick of #18436: upgrade etcd-manager and etcd to latest patches - #18437

Merged
k8s-ci-robot merged 3 commits into
kubernetes:release-1.35from
hakman:etcd-3.6.12-1.35
Jun 2, 2026
Merged

k8s-ci-robot merged 3 commits into
kubernetes:release-1.35from
hakman:etcd-3.6.12-1.35

Conversation

@hakman

@hakman hakman commented Jun 2, 2026

Copy link
Copy Markdown
Member

hakman added 3 commits June 2, 2026 19:15
Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@k8s-ci-robot
k8s-ci-robot requested review from ameukam and rifelpet June 2, 2026 16:33
@k8s-ci-robot k8s-ci-robot added this to the v1.35 milestone Jun 2, 2026
@k8s-ci-robot k8s-ci-robot added size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Jun 2, 2026
@hakman

hakman commented Jun 2, 2026

Copy link
Copy Markdown
Member Author

/approve

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 2, 2026
@ameukam

ameukam commented Jun 2, 2026

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 2, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 9c1e25c into kubernetes:release-1.35 Jun 2, 2026
18 checks passed
@hakman
hakman deleted the etcd-3.6.12-1.35 branch June 2, 2026 17:36
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants