Sitelet https://github.com/kubernetes/kops/pull/18429
Skip to content

Automated cherry pick of #18428: cert-manager: upgrade to v1.19.5 and set AWS_REGION for Route53 dns-01 - #18429

Merged
k8s-ci-robot merged 3 commits into
kubernetes:release-1.35from
hakman:automated-cherry-pick-of-#18428-upstream-release-1.35
May 29, 2026
Merged

k8s-ci-robot merged 3 commits into
kubernetes:release-1.35from
hakman:automated-cherry-pick-of-#18428-upstream-release-1.35

Conversation

@hakman

@hakman hakman commented May 29, 2026

Copy link
Copy Markdown
Member

Cherry pick of #18428 on release-1.35.

#18428: cert-manager: upgrade to v1.19.5 and set AWS_REGION for Route53 dns-01

For details on the cherry pick process, see the cherry pick requests page.

What type of PR is this?


hakman added 3 commits May 29, 2026 16:57
Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@k8s-ci-robot k8s-ci-robot added this to the v1.35 milestone May 29, 2026
@k8s-ci-robot k8s-ci-robot added size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels May 29, 2026
@k8s-ci-robot
k8s-ci-robot requested review from olemarkus and zetaab May 29, 2026 14:15
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added lgtm "Looks good to me", indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels May 29, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 9c9c422 into kubernetes:release-1.35 May 29, 2026
18 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/addons cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XL Denotes a PR that changes 500-999 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants