Sitelet https://github.com/kubernetes/kops/pull/18377
Skip to content

Automated cherry pick of #18375: cilium: add flags for bpf-lb-sock and bpf-lb-sock-hostns-only - #18377

Merged
k8s-ci-robot merged 3 commits into
kubernetes:release-1.34from
hakman:automated-cherry-pick-of-#18375-upstream-release-1.34
May 20, 2026
Merged

k8s-ci-robot merged 3 commits into
kubernetes:release-1.34from
hakman:automated-cherry-pick-of-#18375-upstream-release-1.34

Conversation

@hakman

@hakman hakman commented May 20, 2026

Copy link
Copy Markdown
Member

Cherry pick of #18375 on release-1.34.

#18375: cilium: add flags for bpf-lb-sock and bpf-lb-sock-hostns-only

For details on the cherry pick process, see the cherry pick requests page.

What type of PR is this?


jValdron and others added 3 commits May 20, 2026 14:29
Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@k8s-ci-robot k8s-ci-robot added this to the v1.34 milestone May 20, 2026
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels May 20, 2026
@k8s-ci-robot
k8s-ci-robot requested a review from olemarkus May 20, 2026 11:35
@k8s-ci-robot
k8s-ci-robot requested a review from zetaab May 20, 2026 11:35
@hakman

hakman commented May 20, 2026

Copy link
Copy Markdown
Member Author

/cc @rifelpet @ameukam

@k8s-ci-robot
k8s-ci-robot requested review from ameukam and rifelpet May 20, 2026 11:37
@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label May 20, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 20, 2026
@k8s-ci-robot
k8s-ci-robot merged commit a0cb7a2 into kubernetes:release-1.34 May 20, 2026
18 checks passed
@hakman
hakman deleted the automated-cherry-pick-of-#18375-upstream-release-1.34 branch May 22, 2026 09:11
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/addons area/api cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants