Sitelet https://github.com/kubernetes/kops/pull/18319
Skip to content

azure: list VMSS NICs in protokube gossip seed discovery - #18319

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:azure-fix-gossip-discovery
May 12, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:azure-fix-gossip-discovery

Conversation

@hakman

@hakman hakman commented May 12, 2026

Copy link
Copy Markdown
Member

@k8s-ci-robot
k8s-ci-robot requested review from ameukam and rifelpet May 12, 2026 13:41
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels May 12, 2026
@hakman

hakman commented May 12, 2026

Copy link
Copy Markdown
Member Author

/test pull-kops-azure-gossip

@hakman

hakman commented May 12, 2026

Copy link
Copy Markdown
Member Author

/override pull-kops-e2e-k8s-gce-cilium

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-k8s-gce-cilium

Details

In response to this:

/override pull-kops-e2e-k8s-gce-cilium

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label May 12, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 12, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 0535de4 into kubernetes:master May 12, 2026
26 of 27 checks passed
k8s-ci-robot added a commit that referenced this pull request May 12, 2026
…-upstream-release-1.34

Automated cherry pick of #18319: azure: list VMSS NICs in protokube gossip seed discovery
k8s-ci-robot added a commit that referenced this pull request May 12, 2026
…-upstream-release-1.35

Automated cherry pick of #18319: azure: list VMSS NICs in protokube gossip seed discovery
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants