Sitelet https://github.com/kubernetes/kops/pull/18287
Skip to content

Automated cherry pick of #18285: cilium: allow setting arbitrary cilium-config entries - #18287

Merged
k8s-ci-robot merged 1 commit into
kubernetes:release-1.34from
hakman:automated-cherry-pick-of-#18285-upstream-release-1.34
May 8, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:release-1.34from
hakman:automated-cherry-pick-of-#18285-upstream-release-1.34

Conversation

@hakman

@hakman hakman commented May 8, 2026

Copy link
Copy Markdown
Member

Cherry pick of #18285 on release-1.34.

#18285: cilium: allow setting arbitrary cilium-config entries

For details on the cherry pick process, see the cherry pick requests page.

What type of PR is this?


@k8s-ci-robot k8s-ci-robot added this to the v1.34 milestone May 8, 2026
@k8s-ci-robot k8s-ci-robot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label May 8, 2026
@k8s-ci-robot
k8s-ci-robot requested review from olemarkus and zetaab May 8, 2026 05:17
@k8s-ci-robot k8s-ci-robot added area/addons area/api cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels May 8, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added lgtm "Looks good to me", indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels May 8, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 0cf3674 into kubernetes:release-1.34 May 8, 2026
17 checks passed
@hakman
hakman deleted the automated-cherry-pick-of-#18285-upstream-release-1.34 branch May 22, 2026 09:10
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/addons area/api cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants