cilium: require k8s-connectivity in liveness probe - #18237
Conversation
a4e64f1 to
ddea5ec
Compare
ddea5ec to
dd375d0
Compare
|
can you link to any failing jobs or error logs that this would fix? |
not really, will happen only with real dns or gossip. tested it with gossip and was surprised the agents don't pick up control-plane changes, just get stuck forever waiting to be restarted. |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: rifelpet The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/test all |
…-upstream-release-1.35 Automated cherry pick of #18237: cilium: require k8s-connectivity in liveness probe
Reproduced as part of https://prow.k8s.io/view/gs/kubernetes-ci-logs/pr-logs/pull/kops/18244/pull-kops-aws-upgrade-k135-ko135-to-k136-kolatest-many-addons/2047969291231301632. |
…-upstream-release-1.34 Automated cherry pick of #18237: cilium: require k8s-connectivity in liveness probe
This is a regression that blocks refresh of control-plane address for the agents. This was introduced during the last major update and is present in kOps v1.34 and v1.35.
The Q would, do we want to remove this "feature" or we may want to ignore cilium agents during cluster validation?
https://docs.cilium.io/en/stable/api/
cilium/cilium#32724
/cc @rifelpet @ameukam