Sitelet https://github.com/kubernetes/kops/pull/18232
Skip to content

Initial support for Ubuntu 26.04 - #18232

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
upodroid:ubuntu-2604-support
Apr 23, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
upodroid:ubuntu-2604-support

Conversation

@upodroid

Copy link
Copy Markdown
Member

Ubuntu 26.04 is out

@k8s-ci-robot
k8s-ci-robot requested a review from hakman April 23, 2026 15:47
@k8s-ci-robot
k8s-ci-robot requested a review from zetaab April 23, 2026 15:47
@k8s-ci-robot k8s-ci-robot added size/M Denotes a PR that changes 30-99 lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Apr 23, 2026
Comment thread docs/operations/images.md Outdated
@hakman hakman changed the title add support for ubuntu 26.04 Initial support for ubuntu 26.04 Apr 23, 2026
@hakman hakman changed the title Initial support for ubuntu 26.04 Initial support for Ubuntu 26.04 Apr 23, 2026
@upodroid
upodroid force-pushed the ubuntu-2604-support branch from f96c533 to d6cc1be Compare April 23, 2026 16:48

@hakman hakman left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks!

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Apr 23, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 23, 2026
@hakman

hakman commented Apr 23, 2026

Copy link
Copy Markdown
Member

/test all

@rifelpet

Copy link
Copy Markdown
Member

Lets add a test-infra job under kops-distros as well.

@hakman

hakman commented Apr 23, 2026

Copy link
Copy Markdown
Member

/hold for kubernetes/test-infra#36891

@k8s-ci-robot k8s-ci-robot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 23, 2026
@hakman

hakman commented Apr 23, 2026

Copy link
Copy Markdown
Member

/test pull-kops-aws-distro-ubuntu2604

@hakman

hakman commented Apr 23, 2026

Copy link
Copy Markdown
Member

Test passed...
/unhold

@k8s-ci-robot k8s-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 23, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 24a45b3 into kubernetes:master Apr 23, 2026
27 checks passed
k8s-ci-robot added a commit that referenced this pull request May 10, 2026
…32-origin-release-1.35

Automated cherry pick of #18232: Initial support for Ubuntu 26.04
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/documentation cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants