Sitelet https://github.com/kubernetes/kops/pull/18216
Skip to content

Automated cherry pick of #17877: truncate very long sqs queue names #17885: fix truncate long sqs - #18216

Merged
k8s-ci-robot merged 2 commits into
kubernetes:release-1.34from
hakman:automated-cherry-pick-of-#17877-#17885-upstream-release-1.34
Apr 18, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:release-1.34from
hakman:automated-cherry-pick-of-#17877-#17885-upstream-release-1.34

Conversation

@hakman

@hakman hakman commented Apr 18, 2026

Copy link
Copy Markdown
Member

Cherry pick of #17877 #17885 on release-1.34.

#17877: truncate very long sqs queue names
#17885: fix truncate long sqs

For details on the cherry pick process, see the cherry pick requests page.


Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@k8s-ci-robot k8s-ci-robot added this to the v1.34 milestone Apr 18, 2026
@k8s-ci-robot k8s-ci-robot added the size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. label Apr 18, 2026
@hakman

hakman commented Apr 18, 2026

Copy link
Copy Markdown
Member Author

/cc @rifelpet

@k8s-ci-robot
k8s-ci-robot requested a review from rifelpet April 18, 2026 17:35
@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Apr 18, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 18, 2026
@hakman

hakman commented Apr 18, 2026

Copy link
Copy Markdown
Member Author

/check-cla
/easycla

@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Apr 18, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 7f31a99 into kubernetes:release-1.34 Apr 18, 2026
23 of 24 checks passed
@hakman
hakman deleted the automated-cherry-pick-of-#17877-#17885-upstream-release-1.34 branch April 18, 2026 19:01
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XL Denotes a PR that changes 500-999 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants