Sitelet https://github.com/kubernetes/kops/pull/18209
Skip to content

Automated cherry pick of #18114: Pin upload-artifacts GHA to commit sha - #18209

Merged
k8s-ci-robot merged 1 commit into
kubernetes:release-1.34from
rifelpet:automated-cherry-pick-of-#18114-origin-release-1.34
Apr 17, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:release-1.34from
rifelpet:automated-cherry-pick-of-#18114-origin-release-1.34

Conversation

@rifelpet

Copy link
Copy Markdown
Member

Cherry pick of #18114 on release-1.34.

#18114: Pin upload-artifacts GHA to commit sha

For details on the cherry pick process, see the cherry pick requests page.

What type of PR is this?


@k8s-ci-robot k8s-ci-robot added this to the v1.34 milestone Apr 17, 2026
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Apr 17, 2026
@k8s-ci-robot k8s-ci-robot added lgtm "Looks good to me", indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Apr 17, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot merged commit 29de6ae into kubernetes:release-1.34 Apr 17, 2026
17 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants