Sitelet https://github.com/kubernetes/kops/pull/18206
Skip to content

Automated cherry pick of #18173: Update Golang to 1.25.9 - #18206

Merged
k8s-ci-robot merged 3 commits into
kubernetes:release-1.34from
hakman:automated-cherry-pick-of-#18173-upstream-release-1.34
Apr 18, 2026
Merged

k8s-ci-robot merged 3 commits into
kubernetes:release-1.34from
hakman:automated-cherry-pick-of-#18173-upstream-release-1.34

Conversation

@hakman

@hakman hakman commented Apr 17, 2026

Copy link
Copy Markdown
Member

Cherry pick of #18173 on release-1.34.

#18173: Update Golang to 1.25.9 Signed-off-by: Arnaud Meukam

For details on the cherry pick process, see the cherry pick requests page.


@k8s-ci-robot k8s-ci-robot added this to the v1.34 milestone Apr 17, 2026
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Apr 17, 2026
@hakman hakman changed the title Automated cherry pick of #18173: Update Golang to 1.25.9 Signed-off-by: Arnaud Meukam Automated cherry pick of #18173: Update Golang to 1.25.9 Apr 17, 2026
@hakman
hakman force-pushed the automated-cherry-pick-of-#18173-upstream-release-1.34 branch from 478dafd to 2303c4a Compare April 17, 2026 17:04
@hakman

hakman commented Apr 17, 2026

Copy link
Copy Markdown
Member Author

/cc @rifelpet

@k8s-ci-robot
k8s-ci-robot requested a review from rifelpet April 17, 2026 17:04
@k8s-ci-robot k8s-ci-robot added lgtm "Looks good to me", indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Apr 17, 2026
Signed-off-by: Arnaud Meukam <ameukam@gmail.com>
Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@hakman
hakman force-pushed the automated-cherry-pick-of-#18173-upstream-release-1.34 branch from 2303c4a to e5f539f Compare April 17, 2026 18:58
@k8s-ci-robot k8s-ci-robot removed the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Apr 17, 2026
Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@hakman
hakman force-pushed the automated-cherry-pick-of-#18173-upstream-release-1.34 branch from 2ef9624 to 1a378bc Compare April 17, 2026 19:03
@hakman
hakman requested a review from rifelpet April 17, 2026 19:04
Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@hakman
hakman force-pushed the automated-cherry-pick-of-#18173-upstream-release-1.34 branch from 1a378bc to 351ba92 Compare April 17, 2026 23:14
@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Apr 17, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot merged commit b0b66ca into kubernetes:release-1.34 Apr 18, 2026
18 checks passed
@hakman
hakman deleted the automated-cherry-pick-of-#18173-upstream-release-1.34 branch May 22, 2026 09:10
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants