Sitelet https://github.com/kubernetes/kops/pull/18128
Skip to content

vfs: Silence warnings when the S3 provider has no supported checksum - #18128

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:s3-checksum
Mar 29, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:s3-checksum

Conversation

@hakman

@hakman hakman commented Mar 29, 2026 •

Copy link
Copy Markdown
Member

The fix silences the warning when the S3 provider has no supported checksum, leaving it enabled for AWS.
This is more common than not and can be seen for DO, Hetzner, and possibly Linode/Akamai.

% kops --name my.k8s export kubeconfig --admin
SDK 2026/03/29 12:30:27 WARN Response has no supported checksum. Not validating response payload.
SDK 2026/03/29 12:30:27 WARN Response has no supported checksum. Not validating response payload.
kOps has set your kubectl context to my.k8s

/cc @rifelpet @moshevayner @justinsb

Ref: #18105

Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@k8s-ci-robot k8s-ci-robot added size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Mar 29, 2026
@hakman

hakman commented Mar 29, 2026

Copy link
Copy Markdown
Member Author

I might add some more logic here.
/hold

@k8s-ci-robot k8s-ci-robot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Mar 29, 2026
@hakman

hakman commented Mar 29, 2026

Copy link
Copy Markdown
Member Author

Will continue with a separate PR for the rest.
/unhold

@k8s-ci-robot k8s-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Mar 29, 2026
@hakman

hakman commented Mar 29, 2026

Copy link
Copy Markdown
Member Author

/approve

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 29, 2026
@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Mar 29, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman, rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot merged commit eb186cb into kubernetes:master Mar 29, 2026
26 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Mar 29, 2026
@hakman
hakman deleted the s3-checksum branch March 29, 2026 14:34
k8s-ci-robot added a commit that referenced this pull request Mar 29, 2026
…-upstream-release-1.35

Automated cherry pick of #18128: vfs: Silence warnings when the S3 provider has no supported
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/provider/akamai area/provider/linode cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants