Sitelet https://github.com/kubernetes/kops/pull/18100
Skip to content

[aiconformance] add test for schedulingOrchestration clusterAutoscaling - #18100

Merged
k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
justinsb:aiconformance_schedulingorchestration_clusterautoscaling
Mar 20, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
justinsb:aiconformance_schedulingorchestration_clusterautoscaling

Conversation

@justinsb

@justinsb justinsb commented Mar 19, 2026 •

Copy link
Copy Markdown
Member
  • [aiconformance] add test for schedulingOrchestration clusterAutoscaling

@k8s-ci-robot k8s-ci-robot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 19, 2026
@k8s-ci-robot k8s-ci-robot added size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Mar 19, 2026
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-e2e-k8s-aws-calico

@justinsb
justinsb force-pushed the aiconformance_schedulingorchestration_clusterautoscaling branch from 1df92a3 to 63a0a9a Compare March 19, 2026 16:28
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb
justinsb force-pushed the aiconformance_schedulingorchestration_clusterautoscaling branch from 63a0a9a to 29aa401 Compare March 19, 2026 20:47
@k8s-ci-robot k8s-ci-robot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. and removed size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. labels Mar 19, 2026
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

…akes in CI

Unclear if this is a real issue or just a flake in CI, but increasing the timeout to 10m should help avoid flakes.

We can/should reduce it in another PR and try to root-cause.
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-kubernetes-e2e-ubuntu-gce-build

@justinsb

Copy link
Copy Markdown
Member Author

/assign @hakman @ameukam

@justinsb justinsb changed the title WIP: [aiconformance] add test for schedulingOrchestration clusterAutoscaling [aiconformance] add test for schedulingOrchestration clusterAutoscaling Mar 20, 2026
@k8s-ci-robot k8s-ci-robot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 20, 2026
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-kubernetes-e2e-ubuntu-gce-build

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@justinsb: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
pull-kops-kubernetes-e2e-ubuntu-gce-build b6a2833 link false /test pull-kops-kubernetes-e2e-ubuntu-gce-build

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Mar 20, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 20, 2026
@k8s-ci-robot
k8s-ci-robot merged commit e5dd358 into kubernetes:master Mar 20, 2026
27 of 28 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Mar 20, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants