Sitelet https://github.com/kubernetes/kops/pull/18089
Skip to content

Fix existing VPCs/Subnets with IPv6 - #18089

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
rlees85:existing-ipv6
Mar 18, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
rlees85:existing-ipv6

Conversation

@rlees85

@rlees85 rlees85 commented Mar 17, 2026

Copy link
Copy Markdown
Contributor

KOPS supports private subnets that are IPv6 only. Unfortunately, the logic when networkId is set prevents KOPS from working when any subnet does not have cidr set.

See issue: #18078

I could not see any tests for when an existing network is set. I guess because there is no "cloud", for that reason I did not update tests, just ensured existing tests passed.

@k8s-ci-robot
k8s-ci-robot requested review from hakman and zetaab March 17, 2026 21:33
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Mar 17, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

Hi @rlees85. Thanks for your PR.

I'm waiting for a kubernetes member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@hakman

hakman commented Mar 18, 2026

Copy link
Copy Markdown
Member

/ok-to-test

@k8s-ci-robot k8s-ci-robot added ok-to-test Indicates a non-member PR verified by an org member that is safe to test. and removed needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Mar 18, 2026
@hakman

hakman commented Mar 18, 2026

Copy link
Copy Markdown
Member

Thanks @rlees85. Would you mind squashing the commits and using a meaningful commit message?

@hakman hakman changed the title Fix Existing VPCs/Subnets with IPv6 Fix existing VPCs/Subnets with IPv6 Mar 18, 2026
a CIDR range associated

Signed-off-by: Rich <git0@bitservices.io>
@rlees85

rlees85 commented Mar 18, 2026

Copy link
Copy Markdown
Contributor Author

Yep my bad, they were pretty bad. Apologies. All done

@hakman

hakman commented Mar 18, 2026

Copy link
Copy Markdown
Member

Thanks @rlees85, LGTM!
/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Mar 18, 2026
@hakman

hakman commented Mar 18, 2026

Copy link
Copy Markdown
Member

/override pull-kops-e2e-k8s-gce-cilium

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-k8s-gce-cilium

Details

In response to this:

/override pull-kops-e2e-k8s-gce-cilium

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 18, 2026
@hakman

hakman commented Mar 18, 2026

Copy link
Copy Markdown
Member

/test all

@k8s-ci-robot
k8s-ci-robot merged commit 80196b9 into kubernetes:master Mar 18, 2026
26 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Mar 18, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants