[aiconformance]: Write conformance report to file - #18084
Merged
k8s-ci-robot merged 1 commit intoMar 17, 2026
Merged
Conversation
Member
Author
|
/test pull-kops-ai-conformance |
justinsb
force-pushed
the
write_aiconformance_report
branch
from
March 17, 2026 09:41
5200557 to
4f59081
Compare
Member
Author
|
/test pull-kops-ai-conformance |
justinsb
force-pushed
the
write_aiconformance_report
branch
4 times, most recently
from
March 17, 2026 13:01
5eb635f to
176061c
Compare
Member
Author
|
/test pull-kops-ai-conformance |
1 similar comment
Member
Author
|
/test pull-kops-ai-conformance |
justinsb
force-pushed
the
write_aiconformance_report
branch
from
March 17, 2026 14:01
176061c to
1f034e1
Compare
Member
Author
|
/test pull-kops-ai-conformance |
Member
Author
Member
Author
hakman
approved these changes
Mar 17, 2026
Contributor
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: hakman The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
No description provided.