Sitelet https://github.com/kubernetes/kops/pull/18082
Skip to content

scaletest: Align kops scalability configuration with kube-up for CCM concurrentNodeSyncs - #18082

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
ronaldngounou:align-kops-scalability-with-kube-up-concurrentNodeSyncs
Aug 6, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
ronaldngounou:align-kops-scalability-with-kube-up-concurrentNodeSyncs

Conversation

@ronaldngounou

Copy link
Copy Markdown
Member

Align kops scalability configuration with kube-up by setting cloudControllerManager.concurrentNodeSyncs=5 to match the kubernetes default values.

Contributes to #18070

@k8s-ci-robot k8s-ci-robot added size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Mar 16, 2026
@ronaldngounou

Copy link
Copy Markdown
Member Author

/test pull-kops-gce-master-scale-performance-100
/test pull-kops-ec2-master-scale-performance-100

@ronaldngounou
ronaldngounou force-pushed the align-kops-scalability-with-kube-up-concurrentNodeSyncs branch from 730556b to 0c6cc15 Compare March 16, 2026 21:51
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign justinsb for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@hakman hakman changed the title Align kops scalability configuration with kube-up for CCM concurrentNodeSyncs scaletest: Align kops scalability configuration with kube-up for CCM concurrentNodeSyncs Mar 18, 2026
@ronaldngounou

Copy link
Copy Markdown
Member Author

/test pull-kops-e2e-k8s-gce-cilium

@k8s-ci-robot k8s-ci-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label May 20, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@serathius

Copy link
Copy Markdown
Member

Can you rebase the PR?

…odeSyncs

Signed-off-by: ronaldngounou <rngounou@amazon.com>
@ronaldngounou
ronaldngounou force-pushed the align-kops-scalability-with-kube-up-concurrentNodeSyncs branch from 0c6cc15 to 654f0ff Compare July 31, 2026 08:12
@kubernetes-prow kubernetes-prow Bot removed the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Jul 31, 2026
@ronaldngounou

Copy link
Copy Markdown
Member Author

PR rebased

@serathius

Copy link
Copy Markdown
Member

/lgtm
/approve

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Aug 5, 2026
@hakman

hakman commented Aug 6, 2026

Copy link
Copy Markdown
Member

/approve

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman, serathius

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 6, 2026
@hakman

hakman commented Aug 6, 2026

Copy link
Copy Markdown
Member

/override pull-kops-e2e-k8s-aws-calico

@kubernetes-prow

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-k8s-aws-calico

Details

In response to this:

/override pull-kops-e2e-k8s-aws-calico

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@kubernetes-prow
kubernetes-prow Bot merged commit d11399b into kubernetes:master Aug 6, 2026
28 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/api cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants