Sitelet https://github.com/kubernetes/kops/pull/18080
Skip to content

scaletest: Align kops scalability configuration with kube-up for scheduler QPS/BURST - #18080

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
ronaldngounou:align-kops-scalability-with-kube-up-scheduler-qps-burst
Aug 5, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
ronaldngounou:align-kops-scalability-with-kube-up-scheduler-qps-burst

Conversation

@ronaldngounou

Copy link
Copy Markdown
Member

Align kops scalability configuration with kube-up for scheduler QPS/BURST by setting KOPS_SCHEDULER_QPS/BURST = 100

Contributes to #18070

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign hakuna-matatah for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot requested review from dims and hakman March 16, 2026 21:27
@k8s-ci-robot k8s-ci-robot added size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Mar 16, 2026
@ronaldngounou

Copy link
Copy Markdown
Member Author

/test pull-kops-gce-master-scale-performance-100
/test pull-kops-ec2-master-scale-performance-100

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@ronaldngounou: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
pull-kops-e2e-k8s-gce-cilium 7a1ae49 link true /test pull-kops-e2e-k8s-gce-cilium
pull-kops-e2e-k8s-gce-ipalias 7a1ae49 link true /test pull-kops-e2e-k8s-gce-ipalias

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@ronaldngounou

ronaldngounou commented Mar 16, 2026 •

Copy link
Copy Markdown
Member Author

https://prow.k8s.io/view/gs/kubernetes-ci-logs/pr-logs/pull/kops/18080/pull-kops-e2e-k8s-gce-cilium/2033656507173703680

E0316 22:06:31.366168   49815 op.go:137] GCE operation failed: googleapi: Error 400: 

The instance_group_manager resource 'projects/k8s-infra-e2e-boskos-059/zones/us-central1-f/

instanceGroupManagers/f-control-plane-us-central1-f-e2e-pr18080-pull-kops-e2e--u0bl43' is already being used by 

'projects/k8s-infra-e2e-boskos-059/regions/us-central1/backendServices/api-e2e-pr18080-pull-kops-e2e-k8s-gce-cilium-k8s-local'

@hakman hakman changed the title Align kops scalability configuration with kube-up for scheduler QPS/BURST scaletest: Align kops scalability configuration with kube-up for scheduler QPS/BURST Mar 18, 2026
@k8s-triage-robot

Copy link
Copy Markdown

The Kubernetes project currently lacks enough contributors to adequately respond to all PRs.

This bot triages PRs according to the following rules:

  • After 90d of inactivity, lifecycle/stale is applied
  • After 30d of inactivity since lifecycle/stale was applied, lifecycle/rotten is applied
  • After 30d of inactivity since lifecycle/rotten was applied, the PR is closed

You can:

  • Mark this PR as fresh with /remove-lifecycle stale
  • Close this PR with /close
  • Offer to help out with Issue Triage

Please send feedback to sig-contributor-experience at kubernetes/community.

/lifecycle stale

@k8s-ci-robot k8s-ci-robot added the lifecycle/stale Denotes an issue or PR has remained open with no activity and has become stale. label Jun 16, 2026
@k8s-triage-robot

Copy link
Copy Markdown

The Kubernetes project currently lacks enough active contributors to adequately respond to all PRs.

This bot triages PRs according to the following rules:

  • After 90d of inactivity, lifecycle/stale is applied
  • After 30d of inactivity since lifecycle/stale was applied, lifecycle/rotten is applied
  • After 30d of inactivity since lifecycle/rotten was applied, the PR is closed

You can:

  • Mark this PR as fresh with /remove-lifecycle rotten
  • Close this PR with /close
  • Offer to help out with Issue Triage

Please send feedback to sig-contributor-experience at kubernetes/community.

/lifecycle rotten

@kubernetes-prow kubernetes-prow Bot added lifecycle/rotten Denotes an issue or PR that has aged beyond stale and will be auto-closed. and removed lifecycle/stale Denotes an issue or PR has remained open with no activity and has become stale. labels Jul 16, 2026
@serathius

Copy link
Copy Markdown
Member

Can you rebase the PR?

…URST

Signed-off-by: ronaldngounou <rngounou@amazon.com>
@ronaldngounou
ronaldngounou force-pushed the align-kops-scalability-with-kube-up-scheduler-qps-burst branch from 7a1ae49 to f552019 Compare July 31, 2026 08:11
@ronaldngounou

Copy link
Copy Markdown
Member Author

PR rebased

@serathius

Copy link
Copy Markdown
Member

/lgtm
/approve

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Aug 5, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: serathius

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 5, 2026
@hakman

hakman commented Aug 5, 2026

Copy link
Copy Markdown
Member

/override pull-kops-e2e-k8s-aws-calico

@kubernetes-prow

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-k8s-aws-calico

Details

In response to this:

/override pull-kops-e2e-k8s-aws-calico

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@kubernetes-prow
kubernetes-prow Bot merged commit 2ad809c into kubernetes:master Aug 5, 2026
28 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. lifecycle/rotten Denotes an issue or PR that has aged beyond stale and will be auto-closed. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants