Sitelet https://github.com/kubernetes/kops/pull/18077
Skip to content

[aiconformance]: render markdown to HTML - #18077

Merged
k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
justinsb:aiconformance_html_output
Mar 17, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
justinsb:aiconformance_html_output

Conversation

@justinsb

Copy link
Copy Markdown
Member

No description provided.

@k8s-ci-robot k8s-ci-robot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 16, 2026
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Mar 16, 2026
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

1 similar comment
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@ameukam

ameukam commented Mar 16, 2026

Copy link
Copy Markdown
Member

@justinsb do we want to consider a dedicated page on https://kops.sigs.k8s.io/ that will explain how to access the test results when we do the submission to the AI conformance ?

@justinsb

Copy link
Copy Markdown
Member Author

do we want to consider a dedicated page on https://kops.sigs.k8s.io/ that will explain how to access the test results when we do the submission to the AI conformance ?

I want to start by getting us approved as AI Conformant based on the test results. But then yes, let's repeat it as often as we can!

@k8s-ci-robot k8s-ci-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Mar 16, 2026
@justinsb
justinsb force-pushed the aiconformance_html_output branch from 9d33a44 to 7089cd1 Compare March 16, 2026 19:12
@k8s-ci-robot k8s-ci-robot removed the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Mar 16, 2026
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb justinsb changed the title WIP: [aiconformance]: render markdown to HTML [aiconformance]: render markdown to HTML Mar 16, 2026
@k8s-ci-robot k8s-ci-robot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 16, 2026
@justinsb

Copy link
Copy Markdown
Member Author

The HTML is not perfect, but it is much better than raw markdown!

@ameukam

ameukam commented Mar 16, 2026

Copy link
Copy Markdown
Member

Nice!
/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Mar 16, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 17, 2026
@k8s-ci-robot
k8s-ci-robot merged commit fa6bc4a into kubernetes:master Mar 17, 2026
28 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Mar 17, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants