Sitelet https://github.com/kubernetes/kops/pull/18073
Skip to content

[aiconformance]: Add K8s wrapper for DaemonSet object - #18073

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
ameukam:aiconformance-kubeobjects-wrappers
Mar 15, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
ameukam:aiconformance-kubeobjects-wrappers

Conversation

@ameukam

@ameukam ameukam commented Mar 15, 2026

Copy link
Copy Markdown
Member

Similar to the Service wrapper, adding a DaemonSet wrapper. Needed for runtime validation

Written with the help of GH Copilot

Similar to the Service wrapper, adding a DaemonSet wrapper.
Needed for runtime validation

Written with the help of GH Copilot

Signed-off-by: Arnaud Meukam <ameukam@gmail.com>
@k8s-ci-robot k8s-ci-robot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Mar 15, 2026
@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Mar 15, 2026
@ameukam

ameukam commented Mar 15, 2026

Copy link
Copy Markdown
Member Author

/assign @justinsb @hakman

@justinsb

Copy link
Copy Markdown
Member

/test pull-kops-e2e-k8s-gce-cilium

/approve
/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Mar 15, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: justinsb

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 15, 2026
@ameukam

ameukam commented Mar 15, 2026

Copy link
Copy Markdown
Member Author

/retest

@k8s-ci-robot
k8s-ci-robot merged commit e250338 into kubernetes:master Mar 15, 2026
27 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Mar 15, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants