Sitelet https://github.com/kubernetes/kops/pull/18067
Skip to content

[aiconformance]: Enable Metrics Server - #18067

Merged
k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
ameukam:ai-conformance-install-metrics-server
Mar 14, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
ameukam:ai-conformance-install-metrics-server

Conversation

@ameukam

@ameukam ameukam commented Mar 14, 2026

Copy link
Copy Markdown
Member

No description provided.

Signed-off-by: Arnaud Meukam <ameukam@gmail.com>
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@k8s-ci-robot k8s-ci-robot added do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Mar 14, 2026
@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Mar 14, 2026
@ameukam

ameukam commented Mar 14, 2026

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

Signed-off-by: Arnaud Meukam <ameukam@gmail.com>
@ameukam

ameukam commented Mar 14, 2026

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@ameukam
ameukam marked this pull request as ready for review March 14, 2026 10:30
@k8s-ci-robot k8s-ci-robot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 14, 2026
@k8s-ci-robot
k8s-ci-robot requested a review from zetaab March 14, 2026 10:30
@ameukam

ameukam commented Mar 14, 2026

Copy link
Copy Markdown
Member Author

/assign @justinsb @hakman

@justinsb

Copy link
Copy Markdown
Member

/approve
/lgtm

We're getting there :-)

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Mar 14, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: justinsb

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 14, 2026
@hakman hakman changed the title ai-conformance: Enable Metrics Server [aiconformance]: Enable Metrics Server Mar 14, 2026
@k8s-ci-robot
k8s-ci-robot merged commit c58890d into kubernetes:master Mar 14, 2026
28 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Mar 14, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants