Sitelet https://github.com/kubernetes/kops/pull/18063
Skip to content

[aiconformance]: Validate networking/ai_inference requirements - #18063

Merged
k8s-ci-robot merged 5 commits into
kubernetes:masterfrom
justinsb:aiconformance_gateway_api
Mar 14, 2026
Merged

k8s-ci-robot merged 5 commits into
kubernetes:masterfrom
justinsb:aiconformance_gateway_api

Conversation

@justinsb

@justinsb justinsb commented Mar 13, 2026 •

Copy link
Copy Markdown
Member
  • [aiconformance]: wait for cert-manager to be ready before proceeding
  • [aiconformance]: Install kuberay from helm
  • [aiconformance]: also dump pod logs
  • [aiconformance] Validate networking/ai_inference requirements
  • [aiconformance]: kueue queue-name should be a label

@k8s-ci-robot k8s-ci-robot added do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. labels Mar 13, 2026
@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Mar 13, 2026
@justinsb
justinsb force-pushed the aiconformance_gateway_api branch from e5cdf2e to 4e4fb3d Compare March 13, 2026 13:11
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

2 similar comments
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb
justinsb force-pushed the aiconformance_gateway_api branch from ecb1be0 to d66921a Compare March 13, 2026 18:33
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

1 similar comment
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb justinsb changed the title WIP: [aiconformance] Validate networking/ai_inference requirements [aiconformance] Validate networking/ai_inference requirements Mar 14, 2026
@k8s-ci-robot k8s-ci-robot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 14, 2026
@justinsb

Copy link
Copy Markdown
Member Author

/assign @ameukam @hakman

@hakman

hakman commented Mar 14, 2026

Copy link
Copy Markdown
Member

/approve

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 14, 2026
@justinsb

Copy link
Copy Markdown
Member Author

Weird ... I don't have lgtm automatically on this one. Do you mind adding @hakman ?

@ameukam

ameukam commented Mar 14, 2026

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Mar 14, 2026
@hakman

hakman commented Mar 14, 2026

Copy link
Copy Markdown
Member

/retest

@hakman

hakman commented Mar 14, 2026

Copy link
Copy Markdown
Member

/override pull-kops-e2e-k8s-aws-calico
/override pull-kops-e2e-k8s-gce-ipalias

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-k8s-aws-calico, pull-kops-e2e-k8s-gce-ipalias

Details

In response to this:

/override pull-kops-e2e-k8s-aws-calico
/override pull-kops-e2e-k8s-gce-ipalias

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@hakman hakman changed the title [aiconformance] Validate networking/ai_inference requirements [aiconformance]: Validate networking/ai_inference requirements Mar 14, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 9a8c64f into kubernetes:master Mar 14, 2026
28 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Mar 14, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XL Denotes a PR that changes 500-999 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants