Sitelet https://github.com/kubernetes/kops/pull/18055
Skip to content

[aiconformance] Add tests for security / isolation - #18055

Merged
k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
justinsb:aiconformance_security
Mar 13, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
justinsb:aiconformance_security

Conversation

@justinsb

Copy link
Copy Markdown
Member

No description provided.

@k8s-ci-robot k8s-ci-robot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 12, 2026
@k8s-ci-robot k8s-ci-robot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Mar 12, 2026
@justinsb
justinsb force-pushed the aiconformance_security branch from 108b4d7 to 6ac581b Compare March 12, 2026 11:43
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb
justinsb force-pushed the aiconformance_security branch from 6ac581b to 191cff2 Compare March 12, 2026 12:54
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb
justinsb force-pushed the aiconformance_security branch from 191cff2 to 6cbb9e8 Compare March 12, 2026 14:14
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb
justinsb force-pushed the aiconformance_security branch from 6cbb9e8 to 1f0ef59 Compare March 12, 2026 14:56
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

1 similar comment
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb
justinsb force-pushed the aiconformance_security branch from 83bdeee to 9246188 Compare March 12, 2026 16:47
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb
justinsb force-pushed the aiconformance_security branch from 9246188 to 56b489a Compare March 12, 2026 17:56
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb
justinsb force-pushed the aiconformance_security branch from 56b489a to 4714404 Compare March 12, 2026 18:42
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb
justinsb force-pushed the aiconformance_security branch from 4714404 to f111a25 Compare March 12, 2026 19:31
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb
justinsb force-pushed the aiconformance_security branch from f111a25 to 107fc9b Compare March 12, 2026 23:44
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb
justinsb force-pushed the aiconformance_security branch from 107fc9b to 444ed1b Compare March 13, 2026 01:14
@justinsb
justinsb force-pushed the aiconformance_security branch from 444ed1b to 62ee4be Compare March 13, 2026 01:15
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb justinsb changed the title WIP: [aiconformance] Add tests for security / isolation [aiconformance] Add tests for security / isolation Mar 13, 2026
@k8s-ci-robot k8s-ci-robot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 13, 2026
@justinsb

Copy link
Copy Markdown
Member Author

Finally passing cc @hakman / @ameukam

@hakman

hakman commented Mar 13, 2026

Copy link
Copy Markdown
Member

/approve

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 13, 2026
@ameukam

ameukam commented Mar 13, 2026

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Mar 13, 2026
@k8s-ci-robot
k8s-ci-robot merged commit b494e18 into kubernetes:master Mar 13, 2026
28 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Mar 13, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants