Sitelet https://github.com/kubernetes/kops/pull/18033
Skip to content

chore(channels): promote alpha to stable - #18033

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
moshevayner:promote-18020
Mar 4, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
moshevayner:promote-18020

Conversation

@moshevayner

Copy link
Copy Markdown
Member

What this PR does / why we need it:

Promotes #18020

Which issue(s) this PR fixes (optional, in fixes #<issue number>(, fixes #<issue_number>, ...) format, will close the issue(s) when PR gets merged):

Special notes for your reviewer:

@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Mar 4, 2026
@k8s-ci-robot
k8s-ci-robot requested a review from johngmyers March 4, 2026 00:53
@k8s-ci-robot
k8s-ci-robot requested a review from zetaab March 4, 2026 00:53
Signed-off-by: Moshe Vayner <mvayner@akamai.com>
@moshevayner

Copy link
Copy Markdown
Member Author

/retest

@hakman

hakman commented Mar 4, 2026

Copy link
Copy Markdown
Member

/test pull-kops-e2e-k8s-aws-amazonvpc

@hakman hakman left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

As always, awesome work. Thanks!

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Mar 4, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 4, 2026
@hakman

hakman commented Mar 4, 2026

Copy link
Copy Markdown
Member

/override pull-kops-e2e-k8s-aws-amazonvpc

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-k8s-aws-amazonvpc

Details

In response to this:

/override pull-kops-e2e-k8s-aws-amazonvpc

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-ci-robot
k8s-ci-robot merged commit d46e163 into kubernetes:master Mar 4, 2026
26 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Mar 4, 2026
@moshevayner
moshevayner deleted the promote-18020 branch March 4, 2026 15:26
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/channels cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants