Sitelet https://github.com/kubernetes/kops/pull/18026
Skip to content

chore: Bump etcd-manager to v3.0.20260227 - #18026

Merged
k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
hakman:etcd-manager-3.0.20260227
Feb 27, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
hakman:etcd-manager-3.0.20260227

Conversation

@hakman

@hakman hakman commented Feb 27, 2026

Copy link
Copy Markdown
Member

Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@k8s-ci-robot k8s-ci-robot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Feb 27, 2026
@hakman

hakman commented Feb 27, 2026

Copy link
Copy Markdown
Member Author

/approve

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 27, 2026
@ameukam

ameukam commented Feb 27, 2026

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 27, 2026
@k8s-ci-robot
k8s-ci-robot merged commit fd7047f into kubernetes:master Feb 27, 2026
27 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Feb 27, 2026
@hakman
hakman deleted the etcd-manager-3.0.20260227 branch February 27, 2026 14:13
k8s-ci-robot added a commit that referenced this pull request Feb 28, 2026
…-upstream-release-1.35

Automated cherry pick of #18026: chore: Bump etcd-manager to v3.0.20260227
k8s-ci-robot added a commit that referenced this pull request Feb 28, 2026
…-upstream-release-1.34

Automated cherry pick of #18026: chore: Bump etcd-manager to v3.0.20260227
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants