Sitelet https://github.com/kubernetes/kops/pull/18013
Skip to content

gce: Wait to delete InstanceManagers and InstanceTemplates - #18013

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
Camila-B:delete-ig-long-op-new
Feb 24, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
Camila-B:delete-ig-long-op-new

Conversation

@Camila-B

@Camila-B Camila-B commented Feb 23, 2026 •

Copy link
Copy Markdown
Contributor

/assign @hakman

This commit waits until all related Instances have been deleted before
proceeding with the deletion of InstanceManagers and InstanceTemplates
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Feb 23, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

Hi @Camila-B. Thanks for your PR.

I'm waiting for a kubernetes member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@Camila-B Camila-B changed the title Wait to delete InstanceManagers and InstanceTemplates Wait to delete InstanceManagers and InstanceTemplates on GCE Feb 23, 2026
@k8s-ci-robot k8s-ci-robot added the area/provider/gcp Issues or PRs related to gcp provider label Feb 23, 2026
@k8s-ci-robot
k8s-ci-robot requested a review from zetaab February 23, 2026 23:10
@hakman

hakman commented Feb 24, 2026

Copy link
Copy Markdown
Member

/ok-to-test

@k8s-ci-robot k8s-ci-robot added ok-to-test Indicates a non-member PR verified by an org member that is safe to test. and removed needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Feb 24, 2026
@hakman hakman changed the title Wait to delete InstanceManagers and InstanceTemplates on GCE gce: Wait to delete InstanceManagers and InstanceTemplates Feb 24, 2026
@hakman

hakman commented Feb 24, 2026

Copy link
Copy Markdown
Member

/override pull-kops-e2e-k8s-gce-cilium

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-k8s-gce-cilium

Details

In response to this:

/override pull-kops-e2e-k8s-gce-cilium

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@hakman hakman left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks @Camila-B, LGTM!

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 24, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 24, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 8882f94 into kubernetes:master Feb 24, 2026
26 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Feb 24, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/provider/gcp Issues or PRs related to gcp provider area/rolling-update cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

kops delete instancegroup errors with long operation on GCE

3 participants