Sitelet https://github.com/kubernetes/kops/pull/18008
Skip to content

tests/ai-conformance: update test job for DRA v1 - #18008

Merged
k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
justinsb:aiconformance_manifest
Feb 23, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
justinsb:aiconformance_manifest

Conversation

@justinsb

Copy link
Copy Markdown
Member
  • tests/ai-conformance: use server-side-apply for KubeRay Operator
    The KubeRay manifest is too large to apply with client-side apply,
    so switch to server-side apply.

    We also install cert-manager, which is required for the KubeRay webhooks.

  • tests/ai-conformance: update test job for DRA v1

The KubeRay manifest is too large to apply with client-side apply,
so switch to server-side apply.

We also install cert-manager, which is required for the KubeRay webhooks.
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Feb 23, 2026
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 23, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 23, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 4ad354e into kubernetes:master Feb 23, 2026
28 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Feb 23, 2026
@ameukam

ameukam commented Feb 26, 2026

Copy link
Copy Markdown
Member

Ref:

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants