Sitelet https://github.com/kubernetes/kops/pull/18006
Skip to content

tests/ai-conformance: create cluster with some non-GPU nodes - #18006

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
justinsb:aiconformance_non_gpu_nodes
Feb 22, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
justinsb:aiconformance_non_gpu_nodes

Conversation

@justinsb

Copy link
Copy Markdown
Member

Because GPU nodes are tainted, the cluster won't come up if we
don't have any non-GPU nodes.

Start the cluster with c5.large nodes, and then add GPU nodes.

@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Feb 22, 2026
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb

Copy link
Copy Markdown
Member Author

#18005 shows that even without cluster.spec.containerd.nvidiaGPU.enabled, we still taint the nodes and thus the cluster doesn't validate if we only have GPU nodes.

Given GPU nodes are more expensive, I think it makes sense in real environments to have some non-GPU nodes. (And maybe run GPU nodes as-needed). So let's mirror that in this test!

metadata:
name: gpu-nodes
labels:
kops.k8s.io/cluster: ${CLUSTER_NAME}

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This is why we need to make sure that we export CLUSTER_NAME from kubetest2-kops

if [[ -f "${ENV_FILE}" ]]; then
# shellcheck disable=SC1090
. "${ENV_FILE}"
export KOPS_STATE_STORE

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This is a similar fix for when kubetest2-kops generated KOPS_STATE_STORE, but we encountered an error before sourcing it (e.g. in kops validate cluster)

@k8s-ci-robot k8s-ci-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Feb 22, 2026
Because GPU nodes are tainted, the cluster won't come up if we
don't have any non-GPU nodes.

Start the cluster with c5.large nodes, and then add GPU nodes.
@justinsb
justinsb force-pushed the aiconformance_non_gpu_nodes branch from 9a31f33 to ddba021 Compare February 22, 2026 19:26
@k8s-ci-robot k8s-ci-robot removed the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Feb 22, 2026
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@justinsb: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
pull-kops-ai-conformance ddba021 link false /test pull-kops-ai-conformance

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 22, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 22, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 6e4eef0 into kubernetes:master Feb 22, 2026
27 of 28 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Feb 22, 2026
@ameukam

ameukam commented Feb 26, 2026

Copy link
Copy Markdown
Member

Ref:

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants