Sitelet https://github.com/kubernetes/kops/pull/18004
Skip to content

gce: don't request migration on instance types that don't support it - #18004

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
justinsb:gce_dont_request_migration_if_unsupported
Feb 24, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
justinsb:gce_dont_request_migration_if_unsupported

Conversation

@justinsb

Copy link
Copy Markdown
Member

GPU instance types in particular do not support live migration,
and requesting it causes the instance creation to fail.

GPU instance types in particular do not support live migration,
and requesting it causes the instance creation to fail.
@k8s-ci-robot k8s-ci-robot added the area/provider/gcp Issues or PRs related to gcp provider label Feb 22, 2026
@k8s-ci-robot
k8s-ci-robot requested a review from zetaab February 22, 2026 15:57
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Feb 22, 2026
@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 24, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 24, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 823a194 into kubernetes:master Feb 24, 2026
26 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Feb 24, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/provider/gcp Issues or PRs related to gcp provider cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants