Sitelet https://github.com/kubernetes/kops/pull/17997
Skip to content

Scenario Fixes - #17997

Merged
k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
rifelpet:scenario-kops
Feb 21, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
rifelpet:scenario-kops

Conversation

@rifelpet

Copy link
Copy Markdown
Member
  • Set PATH to include kops binary in upgrade scenario script which is needed to run subsequent kops commands in run-test.sh. Fixes this error:

https://prow.k8s.io/view/gs/kubernetes-ci-logs/logs/e2e-kops-aws-upgrade-k132-ko134-to-k133-kolatest/2024954244909502464

F0220 21:48:03.543882 14120 tester.go:519] failed to run ginkgo tester: error querying cluster from kops get cluster e2e-862e062f13-83206.tests-kops-aws.k8s.io -ojson --full: exec: "kops": executable file not found in $PATH

This will likely go away once we migrate upgrade-ab.sh to use --test=exec (likely next once we get the most recently migrated scenarios to pass)

  • Set KOPS env var for --test=exec subprocesses, allowing the test.sh scripts to run kops commands with "${KOPS}". Fixes this error:

https://prow.k8s.io/view/gs/kubernetes-ci-logs/logs/e2e-kops-aws-aws-load-balancer-controller/2024961542822301696

/home/prow/go/src/k8s.io/kops/tests/e2e/scenarios/aws-lb-controller/test.sh: line 41: KOPS: unbound variable

Also removing a redundant kops download in test.sh scripts now that KOPS will be set.

Fixes this error:

F0220 21:48:03.543882   14120 tester.go:519] failed to run ginkgo tester: error querying cluster from kops get cluster e2e-862e062f13-83206.tests-kops-aws.k8s.io -ojson --full: exec: "kops": executable file not found in $PATH
@k8s-ci-robot k8s-ci-robot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Feb 21, 2026
@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Feb 21, 2026
@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 21, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 21, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 51f73aa into kubernetes:master Feb 21, 2026
27 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Feb 21, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants