Sitelet https://github.com/kubernetes/kops/pull/17994
Skip to content

tests/ai-conformance: don't panic if no zones for AWS - #17994

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
justinsb:tolerate_no_zones_in_kubetest2
Feb 20, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
justinsb:tolerate_no_zones_in_kubetest2

Conversation

@justinsb

Copy link
Copy Markdown
Member

During a --build we may not have any zones configured,
so if we can't get a region from the zones, default to us-east1.

@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Feb 20, 2026
During a --build we may not have any zones configured,
so if we can't get a region from the zones, default to us-east-2.
@justinsb
justinsb force-pushed the tolerate_no_zones_in_kubetest2 branch from aa605c9 to 7eeecf9 Compare February 20, 2026 20:42
@ameukam

ameukam commented Feb 20, 2026

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 20, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 20, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 47c9bd5 into kubernetes:master Feb 20, 2026
27 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Feb 20, 2026
@ameukam

ameukam commented Feb 26, 2026

Copy link
Copy Markdown
Member

Ref:

1 similar comment
@ameukam

ameukam commented Feb 26, 2026

Copy link
Copy Markdown
Member

Ref:

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants