Sitelet https://github.com/kubernetes/kops/pull/17992
Skip to content

tests/ai-conformance: fix syntax for enabling gatewayAPI with cilium - #17992

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
justinsb:aiconform_cilium_syntax
Feb 22, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
justinsb:aiconform_cilium_syntax

Conversation

@justinsb

Copy link
Copy Markdown
Member

No description provided.

@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Feb 20, 2026
@justinsb

Copy link
Copy Markdown
Member Author

/test pull-kops-ai-conformance

@justinsb

Copy link
Copy Markdown
Member Author

/hold

@k8s-ci-robot k8s-ci-robot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Feb 20, 2026
@ameukam

ameukam commented Feb 21, 2026

Copy link
Copy Markdown
Member

/test pull-kops-ai-conformance

@hakman

hakman commented Feb 22, 2026

Copy link
Copy Markdown
Member

The change is correct. It won't fix everything but it's an important fix.
/lgtm
/approve
/unhold

@k8s-ci-robot k8s-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Feb 22, 2026
@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 22, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 22, 2026
@k8s-ci-robot

k8s-ci-robot commented Feb 22, 2026 •

Copy link
Copy Markdown
Contributor

@justinsb: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
pull-kops-ai-conformance 95954eb link false /test pull-kops-ai-conformance

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@hakman

hakman commented Feb 22, 2026

Copy link
Copy Markdown
Member

/override pull-kops-e2e-k8s-gce-cilium

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-k8s-gce-cilium

Details

In response to this:

/override pull-kops-e2e-k8s-gce-cilium

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-ci-robot
k8s-ci-robot merged commit c2f4952 into kubernetes:master Feb 22, 2026
27 of 28 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Feb 22, 2026
@ameukam

ameukam commented Feb 26, 2026

Copy link
Copy Markdown
Member

Ref:

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants