Sitelet https://github.com/kubernetes/kops/pull/17952
Skip to content

Add warning for non-production use in README - #17952

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
ritazh:patch-1
Feb 26, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
ritazh:patch-1

Conversation

@ritazh

@ritazh ritazh commented Feb 13, 2026

Copy link
Copy Markdown
Member

Added a warning about the component's intended use.

Added a warning about the component's intended use.
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Feb 13, 2026
@ritazh

ritazh commented Feb 13, 2026

Copy link
Copy Markdown
Member Author

cc @justinsb

@justinsb justinsb left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve
/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 26, 2026
@justinsb

Copy link
Copy Markdown
Member

Thanks @ritazh

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: justinsb

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 26, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 8dcc550 into kubernetes:master Feb 26, 2026
26 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Feb 26, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants