Sitelet https://github.com/kubernetes/kops/pull/17951
Skip to content

Refactor interface type check to more idiomatic syntax - #17951

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
rramkumar1:interface-refactor
Mar 18, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
rramkumar1:interface-refactor

Conversation

@rramkumar1

@rramkumar1 rramkumar1 commented Feb 13, 2026 •

Copy link
Copy Markdown
Contributor

The end behavior is the same but casting the underlying struct type to a nil pointer is more idiomatic that declaring a pointer to an empty struct.

This isn't something that's explicity mentioned in the Go style guide but has been adopted as a common pattern (e.g https://github.com/uber-go/guide/blob/master/style.md#verify-interface-compliance)

(Note: I did manually edit one file and then prompt an LLM to do the rest for me)

/assign @hakman

@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Feb 13, 2026
@k8s-ci-robot
k8s-ci-robot requested a review from hakman February 13, 2026 00:39
@k8s-ci-robot k8s-ci-robot added area/provider/aws Issues or PRs related to aws provider area/provider/azure Issues or PRs related to azure provider area/provider/gcp Issues or PRs related to gcp provider area/provider/openstack Issues or PRs related to openstack provider size/L Denotes a PR that changes 100-499 lines, ignoring generated files. needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Feb 13, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

Hi @rramkumar1. Thanks for your PR.

I'm waiting for a kubernetes member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-ci-robot k8s-ci-robot added area/provider/digitalocean Issues or PRs related to digitalocean provider area/provider/hetzner Issues or PRs related to Hetzner provider area/provider/scaleway Issues or PRs related to Scaleway provider area/provider/spotinst Issues or PRs related to spotinst provider size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. and removed size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Feb 13, 2026
@rramkumar1

Copy link
Copy Markdown
Contributor Author

@hakman Any thoughts on this PR?

@hakman

hakman commented Mar 18, 2026

Copy link
Copy Markdown
Member

Thanks @rramkumar1, seems like a nice change.
/ok-to-test

@k8s-ci-robot k8s-ci-robot added ok-to-test Indicates a non-member PR verified by an org member that is safe to test. and removed needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Mar 18, 2026
@hakman hakman closed this Mar 18, 2026
@hakman hakman reopened this Mar 18, 2026
@hakman

hakman commented Mar 18, 2026

Copy link
Copy Markdown
Member

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Mar 18, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 18, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@rramkumar1: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
pull-kops-e2e-aws-upgrade-k133-ko133-to-kstable-kolatest-many-addons db91ade link false /test pull-kops-e2e-aws-upgrade-k133-ko133-to-kstable-kolatest-many-addons

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@k8s-ci-robot
k8s-ci-robot merged commit 4b46253 into kubernetes:master Mar 18, 2026
26 of 27 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Mar 18, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/nodeup area/provider/aws Issues or PRs related to aws provider area/provider/azure Issues or PRs related to azure provider area/provider/digitalocean Issues or PRs related to digitalocean provider area/provider/gcp Issues or PRs related to gcp provider area/provider/hetzner Issues or PRs related to Hetzner provider area/provider/openstack Issues or PRs related to openstack provider area/provider/scaleway Issues or PRs related to Scaleway provider area/provider/spotinst Issues or PRs related to spotinst provider cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. size/XL Denotes a PR that changes 500-999 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants