Sitelet https://github.com/kubernetes/kops/pull/17944
Skip to content

metal: Unpin K8s version - #17944

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:metal-k8s-version
Feb 10, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:metal-k8s-version

Conversation

@hakman

@hakman hakman commented Feb 9, 2026

Copy link
Copy Markdown
Member

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@k8s-ci-robot k8s-ci-robot added do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Feb 9, 2026
Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@hakman

hakman commented Feb 10, 2026

Copy link
Copy Markdown
Member Author

/test all

@hakman
hakman marked this pull request as ready for review February 10, 2026 14:46
@k8s-ci-robot k8s-ci-robot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Feb 10, 2026
@hakman

hakman commented Feb 10, 2026

Copy link
Copy Markdown
Member Author

/approve

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 10, 2026
@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 10, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman, rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot merged commit c92a592 into kubernetes:master Feb 10, 2026
27 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Feb 10, 2026
@hakman
hakman deleted the metal-k8s-version branch February 10, 2026 15:49
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants