Sitelet https://github.com/dotnet/dotnet/pull/8092
Skip to content

[release/10.0.4xx] Source code updates from dotnet/scenario-tests - #8092

Merged
dkurepa merged 5 commits into
release/10.0.4xxfrom
darc-release/10.0.4xx-5cb1ae6c-610a-4517-bfce-c278d847117f
Aug 7, 2026
Merged

dkurepa merged 5 commits into
release/10.0.4xxfrom
darc-release/10.0.4xx-5cb1ae6c-610a-4517-bfce-c278d847117f

Conversation

@dotnet-maestro

@dotnet-maestro dotnet-maestro Bot commented Aug 1, 2026 •

Copy link
Copy Markdown
Contributor

Note

This is a codeflow update. It may contain both source code changes from
the source repo
as well as dependency updates. Learn more here.

This pull request brings the following source code changes

From https://github.com/dotnet/scenario-tests

Diff the source with this PR branch
darc vmr diff --name-only https://github.com/dotnet/scenario-tests:4523c6b295293775312d4af7a8a757b0aff018a6..https://github.com/dotnet/dotnet:darc-release/10.0.4xx-5cb1ae6c-610a-4517-bfce-c278d847117f

@dotnet-maestro

dotnet-maestro Bot commented Aug 1, 2026

Copy link
Copy Markdown
Contributor Author

Caution

🚨 Action Required — Conflict detected

A conflict was detected when trying to update this PR with changes from build 325256 of https://github.com/dotnet/scenario-tests/tree/498c8fc665b5b06915b2536f4691b80162a5a8dd.

@dotnet/product-construction please help resolve the conflict in this PR.

The conflicts in the following files need to be manually resolved:

ℹ️ To resolve the conflicts, please follow these steps:

  1. Clone the current repository

    git clone https://github.com/dotnet/dotnet
    cd dotnet
  2. Make sure your darc is up-to-date
    (version 1.1.0-beta.26374.1 or higher)

    # Linux / MacOS
    ./eng/common/darc-init.sh
    # or on Windows
    .\eng\common\darc-init.ps1
  3. Run from repo's git clone and follow the instructions provided by the command to stage the conflict locally

    darc vmr resolve-conflict --subscription f1a8b1e5-be77-4bb5-9647-322154bc4727 

    This should apply the build 325256 with sources from 498c8fc

  4. Resolve the conflicts, commit & push the changes

  5. Once pushed, the Codeflow verification check will turn green.
    If not, a new build might have flown into the PR and you might need to run the command above again.

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@dotnet-maestro

dotnet-maestro Bot commented Aug 1, 2026

Copy link
Copy Markdown
Contributor Author

Important

⚠️ No subscription flowing code in the opposite direction is set up.
Code changes made in the VMR's src/ directory will not flow back to the source repository on branch "release/10.0".

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@dotnet-maestro

dotnet-maestro Bot commented Aug 1, 2026

Copy link
Copy Markdown
Contributor Author

Note

PRs from original repository included in this codeflow update:

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@azure-pipelines

Copy link
Copy Markdown
Azure Pipelines:
Successfully started running 1 pipeline(s).
1 pipeline(s) were filtered out due to trigger conditions.
There may be pipelines that require an authorized user to comment /azp run to run.

@dotnet-maestro

dotnet-maestro Bot commented Aug 3, 2026

Copy link
Copy Markdown
Contributor Author

Note

PRs from original repository included in this codeflow update:

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@dotnet-maestro

dotnet-maestro Bot commented Aug 6, 2026

Copy link
Copy Markdown
Contributor Author

Note

PRs from original repository included in this codeflow update:

💡 You may consult the FAQ for more information or tag @dotnet/prodconsvcs for assistance.

@dkurepa
dkurepa merged commit 32a15c0 into release/10.0.4xx Aug 7, 2026
13 checks passed
@dkurepa
dkurepa deleted the darc-release/10.0.4xx-5cb1ae6c-610a-4517-bfce-c278d847117f branch August 7, 2026 07:56
This was referenced Sep 8, 2026
This was referenced Oct 2, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant