Sitelet https://github.com/alauda/ui/pull/535
Skip to content

fix(drawer): no default config when using component mode - #535

Merged
JounQin merged 4 commits into
masterfrom
fix/drawer
Dec 19, 2023
Merged

JounQin merged 4 commits into
masterfrom
fix/drawer

Conversation

@igauch

@igauch igauch commented Dec 18, 2023

Copy link
Copy Markdown
Contributor

close #534

@changeset-bot

changeset-bot Bot commented Dec 18, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2d83f71

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@alauda/ui Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@igauch
igauch requested a review from JounQin December 18, 2023 10:29
Comment thread src/drawer/drawer.service.ts Outdated
Comment thread src/drawer/drawer.service.ts Outdated
Comment thread src/drawer/drawer.service.ts Outdated
Comment thread src/drawer/drawer.service.ts Outdated
JounQin
JounQin previously approved these changes Dec 19, 2023
@JounQin JounQin changed the title fix: drawer fix(drawer): no default config when using component mode Dec 19, 2023
Comment thread .changeset/stale-dots-speak.md Outdated
JounQin
JounQin previously approved these changes Dec 19, 2023
JounQin
JounQin previously approved these changes Dec 19, 2023
@JounQin
JounQin merged commit 18ab9e9 into master Dec 19, 2023
@JounQin
JounQin deleted the fix/drawer branch December 19, 2023 04:59
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

bug: drawer bugs

2 participants