feat: table support resize column width - #533
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 6b036ca The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
JounQin
reviewed
Dec 15, 2023
JounQin
requested changes
Dec 15, 2023
JounQin
left a comment
Member
There was a problem hiding this comment.
Need more discussion about the necessary of this feature.
fengtianze
reviewed
Dec 15, 2023
Collaborator
igauch
force-pushed
the
feat/table-resizeable
branch
from
December 29, 2023 03:40
e9c66f5 to
02593b3
Compare
fengtianze
reviewed
Dec 29, 2023
JounQin
requested changes
Dec 29, 2023
igauch
force-pushed
the
feat/table-resizeable
branch
2 times, most recently
from
January 2, 2024 06:07
53e74ec to
e41e310
Compare
JounQin
requested changes
Jan 2, 2024
igauch
force-pushed
the
feat/table-resizeable
branch
from
January 3, 2024 03:11
e41e310 to
e30c12e
Compare
JounQin
requested changes
Jan 3, 2024
igauch
force-pushed
the
feat/table-resizeable
branch
from
January 3, 2024 04:20
e30c12e to
51a3266
Compare
JounQin
requested changes
Jan 3, 2024
igauch
force-pushed
the
feat/table-resizeable
branch
2 times, most recently
from
January 3, 2024 08:02
f3017e6 to
19f56c9
Compare
JounQin
requested changes
Jan 3, 2024
igauch
force-pushed
the
feat/table-resizeable
branch
from
January 4, 2024 02:13
19f56c9 to
068f7ca
Compare
Member
|
不要总是用 amend + force push,每次 push 都要重新 review 完整的 PR?应该分 commit 更方便 review ,也能保留修改历史 |
JounQin
requested changes
Jan 4, 2024
JounQin
requested changes
Jan 4, 2024
JounQin
requested changes
Jan 4, 2024
JounQin
reviewed
Jan 4, 2024
JounQin
requested changes
Jan 4, 2024
JounQin
requested changes
Jan 4, 2024
JounQin
reviewed
Jan 4, 2024
JounQin
reviewed
Jan 4, 2024
JounQin
reviewed
Jan 4, 2024
JounQin
reviewed
Jan 8, 2024
JounQin
reviewed
Jan 8, 2024
JounQin
reviewed
Jan 8, 2024
fengtianze
force-pushed
the
feat/table-resizeable
branch
9 times, most recently
from
January 17, 2024 16:22
0fc75e3 to
787e639
Compare
JounQin
reviewed
Jan 17, 2024
JounQin
reviewed
Jan 17, 2024
fengtianze
force-pushed
the
feat/table-resizeable
branch
from
January 18, 2024 02:11
787e639 to
6ce37cd
Compare
fengtianze
force-pushed
the
feat/table-resizeable
branch
from
January 18, 2024 02:48
6ce37cd to
ce26bce
Compare
JounQin
reviewed
Jan 18, 2024
| cleanup(): void { | ||
| if (this.styleSheet) { | ||
| const i = document.adoptedStyleSheets.indexOf(this.styleSheet); | ||
| document.adoptedStyleSheets.splice(i, 1); |
JounQin
reviewed
Jan 18, 2024
| ); | ||
| } | ||
| if (width.endsWith('px')) { | ||
| return parseInt(width.slice(0, -2)); |
Member
There was a problem hiding this comment.
Suggested change
| return parseInt(width.slice(0, -2)); | |
| return Number.parseFloat(width.slice(0, -2)); |
怎么感觉被恢复成最开始的代码了?所有 review 后的变更都没了。。。
JounQin
reviewed
Jan 18, 2024
| if (width.endsWith('px')) { | ||
| return parseInt(width.slice(0, -2)); | ||
| } | ||
| return parseInt(width); |
Member
There was a problem hiding this comment.
Suggested change
| return parseInt(width); | |
| return Number.parseFloat(width); |
fengtianze
force-pushed
the
feat/table-resizeable
branch
from
January 18, 2024 03:04
ce26bce to
b67106f
Compare
JounQin
reviewed
Jan 18, 2024
JounQin
reviewed
Jan 18, 2024
JounQin
reviewed
Jan 18, 2024
JounQin
reviewed
Jan 18, 2024
| this.renderer2.setStyle( | ||
| markLine, | ||
| 'left', | ||
| offset - markLineWidth + 'px', |
fengtianze
force-pushed
the
feat/table-resizeable
branch
from
January 18, 2024 03:32
b67106f to
6b036ca
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.

close #541