Sitelet https://github.com/alauda/ui/pull/512
Skip to content

fix: update i18n pipe on locale change - #512

Merged
JounQin merged 2 commits into
masterfrom
fix/i18n
Oct 7, 2023
Merged

JounQin merged 2 commits into
masterfrom
fix/i18n

Conversation

@JounQin

@JounQin JounQin commented Oct 7, 2023 •

Copy link
Copy Markdown
Member

fix DEVOPS-30413

close #511

@JounQin
JounQin requested a review from zangguodong October 7, 2023 03:36
@changeset-bot

changeset-bot Bot commented Oct 7, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 935ec20

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@alauda/ui Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@JounQin
JounQin merged commit 457e4e5 into master Oct 7, 2023
@JounQin
JounQin deleted the fix/i18n branch October 7, 2023 03:51
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants