Sitelet https://github.com/alauda/ui/pull/488
Skip to content

chore: cleanup deps, fix build typings error - #488

Merged
JounQin merged 4 commits into
masterfrom
chore/deps
Jul 31, 2023
Merged

JounQin merged 4 commits into
masterfrom
chore/deps

Conversation

@JounQin

@JounQin JounQin commented Jul 28, 2023

Copy link
Copy Markdown
Member

No description provided.

@changeset-bot

changeset-bot Bot commented Jul 28, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 8cd187b

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@alauda/ui Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

Comment thread src/input/number-input/number-input.component.ts Outdated
@JounQin
JounQin force-pushed the chore/deps branch 3 times, most recently from 622adbd to 7b90b80 Compare July 28, 2023 13:39
Comment thread src/tabs/tab-group.component.ts
@JounQin
JounQin merged commit c5ba896 into master Jul 31, 2023
@JounQin
JounQin deleted the chore/deps branch July 31, 2023 02:14
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants