Sitelet https://github.com/alauda/ui/pull/484
Skip to content

fix: fix tree hierarchical relationship - #484

Merged
JounQin merged 2 commits into
masterfrom
fix/fix-tree-hierarchical-relationship
Jul 18, 2023
Merged

JounQin merged 2 commits into
masterfrom
fix/fix-tree-hierarchical-relationship

Conversation

@yinshuxun

@yinshuxun yinshuxun commented Jul 12, 2023 •

Copy link
Copy Markdown
Contributor

原来的层级关系不清晰,容易选错: 比如 c、c-1、c-2 应该对齐,c-0、c-0-1、c-0-2 不应该对齐
image
变更之后如下:
image

@changeset-bot

changeset-bot Bot commented Jul 12, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ff34b74

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@alauda/ui Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

Comment thread src/tree-select/tree-node.component.html Outdated
@yinshuxun
yinshuxun force-pushed the fix/fix-tree-hierarchical-relationship branch from 664587e to a111a08 Compare July 12, 2023 03:44
@yinshuxun
yinshuxun requested review from JounQin and fengtianze July 17, 2023 06:56
fengtianze
fengtianze previously approved these changes Jul 17, 2023
@yinshuxun
yinshuxun force-pushed the fix/fix-tree-hierarchical-relationship branch from a111a08 to c4a210c Compare July 17, 2023 07:03
@yinshuxun
yinshuxun force-pushed the fix/fix-tree-hierarchical-relationship branch from c4a210c to 1f23523 Compare July 17, 2023 07:14
@yinshuxun
yinshuxun requested a review from fengtianze July 18, 2023 03:10
fengtianze
fengtianze previously approved these changes Jul 18, 2023
JounQin
JounQin previously approved these changes Jul 18, 2023
@JounQin
JounQin dismissed stale reviews from fengtianze and themself via ff34b74 July 18, 2023 03:24
@JounQin
JounQin enabled auto-merge (squash) July 18, 2023 03:24
@JounQin
JounQin merged commit bda1860 into master Jul 18, 2023
@JounQin
JounQin deleted the fix/fix-tree-hierarchical-relationship branch July 18, 2023 03:27
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants