Sitelet https://github.com/alauda/ui/pull/466
Skip to content

feat: tooltip add animation; fix: autocomplete init position - #466

Merged
fengtianze merged 2 commits into
masterfrom
chore/tooltip
May 8, 2023
Merged

fengtianze merged 2 commits into
masterfrom
chore/tooltip

Conversation

@igauch

@igauch igauch commented Apr 13, 2023 •

Copy link
Copy Markdown
Contributor
  1. fix: autocomplete init position
  2. chore: tooltip add animation

@changeset-bot

changeset-bot Bot commented Apr 13, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 66ba4d1

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@alauda/ui Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@igauch
igauch requested a review from fengtianze April 25, 2023 07:35
@igauch
igauch force-pushed the chore/tooltip branch 2 times, most recently from 9de51f1 to 35a6923 Compare April 26, 2023 10:30
Comment thread src/tooltip/base-tooltip.ts Outdated
@igauch igauch changed the title Chore tooltip feat: tooltip add animation; fix: autocomplete init position Apr 27, 2023
Comment thread src/autocomplete/autocomplete.component.ts
Comment thread src/autocomplete/autocomplete.component.ts
Comment thread src/autocomplete/autocomplete.directive.ts
Comment thread src/date-picker/date-picker/date-picker.component.ts Outdated
Comment thread src/tooltip/base-tooltip.ts
@igauch
igauch force-pushed the chore/tooltip branch 3 times, most recently from 17c0ce6 to df8f048 Compare May 8, 2023 07:29
@fengtianze

Copy link
Copy Markdown
Collaborator

加一下changeset

fengtianze
fengtianze previously approved these changes May 8, 2023
@fengtianze
fengtianze merged commit 74dde67 into master May 8, 2023
@fengtianze
fengtianze deleted the chore/tooltip branch May 8, 2023 08:54
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants